chore: update nx to v23 - #379
Open
tsang-bot[bot] wants to merge 1 commit into
Open
tsang-bot[bot] wants to merge 1 commit into
tsang-bot[bot] wants to merge 1 commit into
Conversation
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
8 times, most recently
from
October 16, 2024 02:21
9536534 to
c707839
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
5 times, most recently
from
October 24, 2024 02:19
768fc97 to
5efe56f
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
5 times, most recently
from
November 1, 2024 02:29
995f389 to
138d0d7
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
8 times, most recently
from
November 9, 2024 13:22
06abda8 to
3205d3f
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
November 12, 2024 02:18
0faf505 to
767c8fe
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
6 times, most recently
from
December 9, 2024 13:32
395a909 to
2730905
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
December 16, 2024 13:32
9697c36 to
392b0f0
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
from
December 21, 2024 02:18
392b0f0 to
ccf2b8f
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
3 times, most recently
from
January 13, 2025 13:26
e469143 to
e306183
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
2 times, most recently
from
January 24, 2025 02:16
7f72c78 to
bd09920
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
3 times, most recently
from
February 3, 2025 13:24
19e81e0 to
de65bb9
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
February 13, 2025 02:18
4cbe4b5 to
23a8e06
Compare
tsang-bot
Bot
force-pushed
the
renovate/major-nx
branch
4 times, most recently
from
February 21, 2025 02:21
bb2157f to
0a9b9d5
Compare
Contributor
Author
|
Contributor
Author
Edited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
19.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.119.8.14→23.2.1Release Notes
nrwl/nx (@nx/devkit)
v23.2.1Compare Source
23.2.1 (2026-09-09)
🚀 Features
🩹 Fixes
❤️ Thank You
v23.2.0Compare Source
23.2.0 (2026-09-02)
🚀 Features
🩹 Fixes
ng-packagrlazily (#36632)nx affected(#36569, #36568)@find(#36526)❤️ Thank You
v23.1.3Compare Source
23.1.3 (2026-08-31)
🩹 Fixes
ng-packagrlazily (#36632)❤️ Thank You
v23.1.2Compare Source
23.1.2 (2026-08-26)
🩹 Fixes
nx affected(#36569, #36568)@find(#36526)❤️ Thank You
v23.1.1Compare Source
23.1.1 (2026-07-30)
🚀 Features
Configuration
📅 Schedule: (in timezone Europe/London)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR has been generated by Mend Renovate CLI.